Skip to content

chore(plugins/container): bump version to 0.7.5 - #1565

Merged
poiana merged 1 commit into
falcosecurity:mainfrom
c2ndev:c2n/chore/container-0.7.5
Sep 22, 2026
Merged

poiana merged 1 commit into
falcosecurity:mainfrom
c2ndev:c2n/chore/container-0.7.5

Conversation

@c2ndev

@c2ndev c2ndev commented Sep 22, 2026

Copy link
Copy Markdown
Contributor

What type of PR is this?

Uncomment one (or more) /kind <> lines:

/kind bug

/kind cleanup

/kind design

/kind documentation

/kind failing-test

/kind feature

Any specific area of the project related to this PR?

Uncomment one (or more) /area <> lines:

/area plugins

/area registry

/area build

/area documentation

/area automation

What this PR does / why we need it:

Bump the container plugin version to 0.7.5.

Which issue(s) this PR fixes:

Special notes for your reviewer:

Prepare the container plugin for the 0.7.5 release.

Signed-off-by: c2ndev <cannarella.dev@gmail.com>
@github-actions

Copy link
Copy Markdown

Rules files suggestions

@leogr

leogr commented Sep 22, 2026

Copy link
Copy Markdown
Member

/meow

@poiana

poiana commented Sep 22, 2026

Copy link
Copy Markdown
Contributor

@leogr: cat image

Details

In response to this:

/meow

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@poiana
poiana merged commit a246c88 into falcosecurity:main Sep 22, 2026
24 checks passed
@poiana

poiana commented Sep 22, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: c2ndev, leogr

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants