chore(deps): bump org.apache.maven.version from 3.9.12 to 3.9.16 - #1055
dependabot[bot] wants to merge 1 commit into
Conversation
Bumps `org.apache.maven.version` from 3.9.12 to 3.9.16. Updates `org.apache.maven:maven-artifact` from 3.9.12 to 3.9.16 Updates `org.apache.maven:maven-compat` from 3.9.12 to 3.9.16 - [Release notes](https://github.com/apache/maven/releases) - [Commits](apache/maven@maven-3.9.12...maven-3.9.16) Updates `org.apache.maven:maven-core` from 3.9.12 to 3.9.16 Updates `org.apache.maven:maven-model` from 3.9.12 to 3.9.16 Updates `org.apache.maven:maven-plugin-api` from 3.9.12 to 3.9.16 - [Release notes](https://github.com/apache/maven/releases) - [Commits](apache/maven@maven-3.9.12...maven-3.9.16) --- updated-dependencies: - dependency-name: org.apache.maven:maven-artifact dependency-version: 3.9.16 dependency-type: direct:production update-type: version-update:semver-patch - dependency-name: org.apache.maven:maven-compat dependency-version: 3.9.16 dependency-type: direct:production update-type: version-update:semver-patch - dependency-name: org.apache.maven:maven-core dependency-version: 3.9.16 dependency-type: direct:production update-type: version-update:semver-patch - dependency-name: org.apache.maven:maven-model dependency-version: 3.9.16 dependency-type: direct:production update-type: version-update:semver-patch - dependency-name: org.apache.maven:maven-plugin-api dependency-version: 3.9.16 dependency-type: direct:production update-type: version-update:semver-patch ... Signed-off-by: dependabot[bot] <support@github.com>
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a eclipse-che member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
Bumps
org.apache.maven.versionfrom 3.9.12 to 3.9.16.Updates
org.apache.maven:maven-artifactfrom 3.9.12 to 3.9.16Updates
org.apache.maven:maven-compatfrom 3.9.12 to 3.9.16Release notes
Sourced from org.apache.maven:maven-compat's releases.
... (truncated)
Commits
2bdd9fd[maven-release-plugin] prepare release maven-3.9.16229e9d7TrimthreadConfigurationto accept input surrounded with spaces7d5fd94Bump org.codehaus.plexus:plexus-classworlds from 2.9.0 to 2.11.0 (#12039)0d100e5[3.9.x] Bump to parent POM 48 (#12024)7ae7935Backport: Maven 3.10.x fixes plugin resolution, by putting user in charge (#1...86fc95bBump commons-io:commons-io from 2.21.0 to 2.22.0 (#11980)029557aBump com.google.guava:guava from 33.5.0-jre to 33.6.0-jre (#11951)b5250f2Bump actions/cache from 5.0.4 to 5.0.5 (#11943)7ef2c23[maven-release-plugin] prepare for next development iteration98b2cdb[maven-release-plugin] prepare release maven-3.9.15Updates
org.apache.maven:maven-corefrom 3.9.12 to 3.9.16Updates
org.apache.maven:maven-modelfrom 3.9.12 to 3.9.16Updates
org.apache.maven:maven-plugin-apifrom 3.9.12 to 3.9.16Release notes
Sourced from org.apache.maven:maven-plugin-api's releases.
... (truncated)
Commits
2bdd9fd[maven-release-plugin] prepare release maven-3.9.16229e9d7TrimthreadConfigurationto accept input surrounded with spaces7d5fd94Bump org.codehaus.plexus:plexus-classworlds from 2.9.0 to 2.11.0 (#12039)0d100e5[3.9.x] Bump to parent POM 48 (#12024)7ae7935Backport: Maven 3.10.x fixes plugin resolution, by putting user in charge (#1...86fc95bBump commons-io:commons-io from 2.21.0 to 2.22.0 (#11980)029557aBump com.google.guava:guava from 33.5.0-jre to 33.6.0-jre (#11951)b5250f2Bump actions/cache from 5.0.4 to 5.0.5 (#11943)7ef2c23[maven-release-plugin] prepare for next development iteration98b2cdb[maven-release-plugin] prepare release maven-3.9.15Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore this major versionwill close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this minor versionwill close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)@dependabot ignore this dependencywill close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)